[pre-commit.ci] pre-commit autoupdate - #540
pre-commit-ci[bot] wants to merge 1 commit into
Conversation
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 00s |
acf9bc8 to
4b485c1
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 13s |
4b485c1 to
89acb37
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 08s |
89acb37 to
8a8a274
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 06s |
8a8a274 to
6f8daa9
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 18s |
6f8daa9 to
f892d40
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 06s |
f892d40 to
067b480
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 09s |
067b480 to
fed4068
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 10s |
fed4068 to
1e7805d
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 07s |
1e7805d to
3b07c31
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 14s |
3b07c31 to
d46e6fe
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 06s |
updates: - [github.com/psf/black-pre-commit-mirror: 26.5.0 → 26.5.1](psf/black-pre-commit-mirror@26.5.0...26.5.1) - [github.com/PyCQA/flake8: 7.3.0 → 7.4.1](PyCQA/flake8@7.3.0...7.4.1) - [github.com/PyCQA/isort: 9.0.0a3 → 9.0.2](PyCQA/isort@9.0.0a3...9.0.2) - [github.com/pre-commit/mirrors-mypy: v2.1.0 → v2.3.1](pre-commit/mirrors-mypy@v2.1.0...v2.3.1) - [github.com/teemtee/tmt.git: 1.73.0 → 1.79.0](https://github.com/teemtee/tmt.git/compare/1.73.0...1.79.0) - [github.com/python-jsonschema/check-jsonschema: 0.37.2 → 0.38.2](python-jsonschema/check-jsonschema@0.37.2...0.38.2) - [github.com/crate-ci/typos: v1 → v1.50.3](crate-ci/typos@v1...v1.50.3)
d46e6fe to
15804b3
Compare
|
Build succeeded. ✔️ pre-commit SUCCESS in 2m 05s |
updates: