Skip to content
This repository was archived by the owner on Mar 17, 2026. It is now read-only.

chore(BA-3894): Remove stale discount proof routes - #2938

Open
stevieraykatz wants to merge 1 commit into
masterfrom
remove-stale-routes
Open

stevieraykatz wants to merge 1 commit into
masterfrom
remove-stale-routes

Conversation

@stevieraykatz

Copy link
Copy Markdown
Contributor

What changed? Why?

Notes to reviewers

How has it been tested?

Have you tested the following pages?

BaseWeb

  • [] base.org
  • [] base.org/names
  • [] base.org/builders
  • [] base.org/ecosystem
  • [] base.org/name/jesse
  • [] base.org/manage-names
  • [] base.org/resources

@linear

linear Bot commented Jan 28, 2026

Copy link
Copy Markdown

@cb-heimdall

cb-heimdall commented Jan 28, 2026 •

Copy link
Copy Markdown
Collaborator

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@vercel

vercel Bot commented Jan 28, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Review Updated (UTC)
base-web Ready Ready Preview, Comment Jan 28, 2026 6:49pm

Request Review

This branch was successfully deployed

1 active deployment
Preview — e4779b5a Deployed Jan 28, 2026 by vercel[bot]
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants